Skip to content

Update Testgrid Prod to v20260901-v0.0.176-4-g734ae2b9 - #1333

Merged
google-oss-prow[bot] merged 1 commit into
GoogleCloudPlatform:mainfrom
google-oss-robot:autobump-prod
Sep 1, 2026
Merged

Update Testgrid Prod to v20260901-v0.0.176-4-g734ae2b9#1333
google-oss-prow[bot] merged 1 commit into
GoogleCloudPlatform:mainfrom
google-oss-robot:autobump-prod

Conversation

@google-oss-robot

Copy link
Copy Markdown
Collaborator

gcr.io/k8s-testgrid/ changes: 4713188...734ae2b (2026‑06‑05 → 2026‑09‑01)

gcr.io/k8s-testgrid/ changes: GoogleCloudPlatform/testgrid@4713188...734ae2b (2026‑06‑05 → 2026‑09‑01)

@michelle192837 michelle192837 left a comment

Copy link
Copy Markdown
Collaborator

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Spot checked the canary instance, it looks like this should be fine.

@google-oss-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: google-oss-robot, michelle192837

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@google-oss-prow
google-oss-prow Bot merged commit 29aa910 into GoogleCloudPlatform:main Sep 1, 2026
9 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants