Maintenance: Update dependency vitest-browser-svelte to v2 - #684
Open
dle-renovate-bot[bot] wants to merge 1 commit into
Open
Maintenance: Update dependency vitest-browser-svelte to v2#684dle-renovate-bot[bot] wants to merge 1 commit into
dle-renovate-bot[bot] wants to merge 1 commit into
Conversation
🦋 Changeset detectedLatest commit: d6307ce The changes in this PR will be included in the next version bump. This PR includes changesets to release 1 package
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
dle-renovate-bot
Bot
force-pushed
the
renovate/vitest-browser-svelte-2.x
branch
from
April 17, 2026 00:32
400eddf to
9fa2b69
Compare
Contributor
Author
Edited/Blocked NotificationRenovate will not automatically rebase this PR, because it does not recognize the last commit author and assumes somebody else may have edited the PR. You can manually request rebase by checking the rebase/retry box above. |
dle-renovate-bot
Bot
force-pushed
the
renovate/vitest-browser-svelte-2.x
branch
4 times, most recently
from
June 17, 2026 12:24
9944a99 to
36b0fd6
Compare
dle-renovate-bot
Bot
force-pushed
the
renovate/vitest-browser-svelte-2.x
branch
2 times, most recently
from
June 23, 2026 01:01
13da3db to
7d3b793
Compare
dle-renovate-bot
Bot
force-pushed
the
renovate/vitest-browser-svelte-2.x
branch
6 times, most recently
from
July 7, 2026 00:41
f7b0a62 to
f02c8f3
Compare
dle-renovate-bot
Bot
force-pushed
the
renovate/vitest-browser-svelte-2.x
branch
from
July 11, 2026 00:35
f02c8f3 to
61e217b
Compare
| datasource | package | from | to | | ---------- | --------------------- | ----- | ----- | | npm | vitest-browser-svelte | 0.1.0 | 2.2.1 |
dle-renovate-bot
Bot
force-pushed
the
renovate/vitest-browser-svelte-2.x
branch
from
July 12, 2026 00:36
61e217b to
d6307ce
Compare
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
This PR contains the following updates:
^0.1.0→^2.2.1Release Notes
vitest-community/vitest-browser-svelte (vitest-browser-svelte)
v2.2.1Compare Source
No significant changes
View changes on GitHub
v2.2.0Compare Source
🚀 Features
wrapperoption - by @mcous in #31 (3715b)View changes on GitHub
v2.1.1Compare Source
🐞 Bug Fixes
@playwright/testout of dependencies - by @ocavue in #26 (ff94b)View changes on GitHub
v2.1.0Compare Source
🚀 Features
View changes on GitHub
v2.0.2Compare Source
💻 Development
🐞 Bug Fixes
container, update to Vitest 4.0.18 to avoid rendering injecteddata-testidattribute.View changes on GitHub
v2.0.1Compare Source
🐞 Bug Fixes
View changes on GitHub
v2.0.0Compare Source
🚨 Breaking Changes
View changes on GitHub
v1.1.0Compare Source
🚀 Features
locatorfromrender- by @sheremet-va (b25e8)🐞 Bug Fixes
View changes on GitHub
v1.0.0Compare Source
🐞 Bug Fixes
View changes on GitHub
Configuration
📅 Schedule: (UTC)
🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.
🔕 Ignore: Close this PR and you won't be reminded about this update again.
This PR has been generated by Mend Renovate.