Skip to content

Bumped production release number for 1.36 - #4988

Merged
eks-distro-bot merged 1 commit into
mainfrom
1.36-9_production_1790889397
Oct 2, 2026
Merged

eks-distro-bot merged 1 commit into
mainfrom
1.36-9_production_1790889397

Conversation

@charles-large

Copy link
Copy Markdown
Member

Bumped production release number for 1.36 to 9.

@charles-large

Copy link
Copy Markdown
Member Author

/approve
/unhold

@eks-distro-bot

Copy link
Copy Markdown
Collaborator

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: charles-large

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@varma-maryala

Copy link
Copy Markdown
Member

/lgtm

@charles-large

Copy link
Copy Markdown
Member Author

/unhold

@eks-distro-bot
eks-distro-bot merged commit a23a09d into main Oct 2, 2026
5 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants