Skip to content

Bumped development release number for 1.32 - #4978

Merged
eks-distro-bot merged 1 commit into
mainfrom
1.32-49_development_1790801593
Sep 30, 2026
Merged

eks-distro-bot merged 1 commit into
mainfrom
1.32-49_development_1790801593

Conversation

@charles-large

Copy link
Copy Markdown
Member

By submitting this pull request, I confirm that you can use, modify, copy, and redistribute this contribution, under the terms of your choice.

@charles-large

Copy link
Copy Markdown
Member Author

/approve
/unhold

@eks-distro-bot

Copy link
Copy Markdown
Collaborator

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: charles-large

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@varma-maryala

Copy link
Copy Markdown
Member

/lgtm

@charles-large

Copy link
Copy Markdown
Member Author

/unhold

@eks-distro-bot
eks-distro-bot merged commit 6f47488 into main Sep 30, 2026
5 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants