Skip to content

Update go.mod files - #4968

Open
eks-distro-pr-bot wants to merge 1 commit into
aws:mainfrom
eks-distro-pr-bot:go-mod-update-main
Open

eks-distro-pr-bot wants to merge 1 commit into
aws:mainfrom
eks-distro-pr-bot:go-mod-update-main

Conversation

@eks-distro-pr-bot

Copy link
Copy Markdown
Contributor

This PR updates the checked in go.mod and go.sum files across all dependency projects to support automated vulnerability scanning.
Click here to view job logs.

/hold

By submitting this pull request, I confirm that you can use, modify, copy, and redistribute this contribution, under the terms of your choice.

@eks-distro-bot

Copy link
Copy Markdown
Collaborator

Hi @eks-distro-pr-bot. Thanks for your PR.

I'm waiting for a aws member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work.

Tip

We noticed you've done this a few times! Consider joining the org to skip this step and gain /lgtm and other bot rights. We recommend asking approvers on your previous PRs to sponsor you.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@eks-distro-pr-bot
eks-distro-pr-bot force-pushed the go-mod-update-main branch 3 times, most recently from 31254d6 to a527898 Compare September 30, 2026 08:26
@charles-large

Copy link
Copy Markdown
Member

/approve
/lgtm
/unhold

@eks-distro-bot

Copy link
Copy Markdown
Collaborator

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: charles-large

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@eks-distro-bot

Copy link
Copy Markdown
Collaborator

@eks-distro-pr-bot: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
coredns-1-31-presubmit a527898 link true /test coredns-1-31-presubmit
coredns-1-32-presubmit a527898 link true /test coredns-1-32-presubmit
coredns-1-37-presubmit a527898 link true /test coredns-1-37-presubmit
coredns-1-33-presubmit a527898 link true /test coredns-1-33-presubmit
coredns-1-34-presubmit a527898 link true /test coredns-1-34-presubmit
coredns-1-36-presubmit a527898 link true /test coredns-1-36-presubmit
coredns-1-35-presubmit a527898 link true /test coredns-1-35-presubmit

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@eks-distro-bot

Copy link
Copy Markdown
Collaborator

New changes are detected. LGTM label has been removed.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants